Skip to content

Create CODEOWNERS - #383

Merged
otc-zuul[bot] merged 1 commit into
mainfrom
CODEOWNERS
Sep 11, 2026
Merged

Create CODEOWNERS#383
otc-zuul[bot] merged 1 commit into
mainfrom
CODEOWNERS

Conversation

@Aloento

@Aloento Aloento commented Sep 10, 2026

Copy link
Copy Markdown
Member

No description provided.

@Aloento Aloento added the gate Merge PR label Sep 11, 2026
@otc-zuul
otc-zuul Bot merged commit 28a4f09 into main Sep 11, 2026
7 checks passed
@otc-zuul
otc-zuul Bot deleted the CODEOWNERS branch September 11, 2026 09:13
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

gate Merge PR

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants