Skip to content

chore: update dependency turbo to v2.10.12 - #274

Merged
notaphplover merged 2 commits into
masterfrom
renovate/turbo-monorepo
Sep 3, 2026
Merged

chore: update dependency turbo to v2.10.12#274
notaphplover merged 2 commits into
masterfrom
renovate/turbo-monorepo

Conversation

@inversify-app

@inversify-app inversify-app Bot commented Aug 24, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Confidence
turbo (source) 2.10.92.10.12 age confidence

Release Notes

vercel/turborepo (turbo)

v2.10.12: Turborepo v2.10.12

Compare Source

What's Changed

Changelog

New Contributors

Full Changelog: vercel/turborepo@v2.10.11...v2.10.12

v2.10.11: Turborepo v2.10.11

Compare Source

What's Changed

Changelog

Full Changelog: vercel/turborepo@v2.10.10...v2.10.11

v2.10.10: Turborepo v2.10.10

Compare Source

What's Changed

Changelog

New Contributors

Full Changelog: vercel/turborepo@v2.10.9...v2.10.10


Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR has been generated by Mend Renovate CLI.

@inversify-app
inversify-app Bot requested a review from notaphplover as a code owner August 24, 2026 00:38
@changeset-bot

changeset-bot Bot commented Aug 24, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 826f93a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 4 packages
Name Type
@inversifyjs/apollo-core Patch
@inversifyjs/apollo-express Patch
@inversifyjs/apollo-fastify Patch
@inversifyjs/apollo-subscription-ws Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@inversify-app
inversify-app Bot force-pushed the renovate/turbo-monorepo branch from 970562a to b3f951a Compare August 31, 2026 02:12
@inversify-app inversify-app Bot changed the title chore: update dependency turbo to v2.10.11 chore: update dependency turbo to v2.10.12 Aug 31, 2026
@notaphplover
notaphplover merged commit b68c8fe into master Sep 3, 2026
6 checks passed
@notaphplover
notaphplover deleted the renovate/turbo-monorepo branch September 3, 2026 22:58
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant